Skip to content

Prevent server sourcemaps from being part of client output - #12757

Merged
ematipico merged 8 commits into
4-legacyfrom
4-sourcemaps
Dec 18, 2024
Merged

ematipico merged 8 commits into
4-legacyfrom
4-sourcemaps

Conversation

@matthewp

Copy link
Copy Markdown
Contributor

Changes

Testing

  • Tests added

Docs

N/A, bug fix

matthewp and others added 5 commits December 16, 2024 15:59
* Revert "fix custom `assetFileNames` issue (#12449)"

This reverts commit e6b8017.

This change caused source maps to be left in the client folder, exposing them into the server.

* Add changeset
* Clean sourcemaps from static output

* Add changeset

* Update test to look at frontend sourcemaps
* Revert "fix custom `assetFileNames` issue (#12449)"

This reverts commit e6b8017.

This change caused source maps to be left in the client folder, exposing them into the server.

* Add changeset
* Clean sourcemaps from static output

* Add changeset

* Update test to look at frontend sourcemaps
@changeset-bot

changeset-bot Bot commented Dec 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b08b8aa

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added the pkg: astro Related to the core `astro` package (scope) label Dec 16, 2024
@ematipico

Copy link
Copy Markdown
Member

Failing tests are due to tailing wind file using __dirname, which causes errors in Node.js. The test failing doesn't affect this PR.

@ematipico
ematipico merged commit d0aaac3 into 4-legacy Dec 18, 2024
@ematipico
ematipico deleted the 4-sourcemaps branch December 18, 2024 11:45
@astrobot-houston astrobot-houston mentioned this pull request Dec 18, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

pkg: astro Related to the core `astro` package (scope)

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants