Update some tests to 3.14.6 - #8167
Conversation
|
Important Review skippedReview was skipped due to path filters ⛔ Files ignored due to path filters (8)
CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including ⚙️ Run configurationConfiguration used: Path: .coderabbit.yml Review profile: CHILL Plan: Pro Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
✨ Finishing Touches🧪 Generate unit tests (beta)
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
📦 Library DependenciesThe following Lib/ modules were modified. Here are their dependencies: [x] test: cpython/Lib/test/test_fileutils.py dependencies: dependent tests: (no tests depend on fileutils) [ ] test: cpython/Lib/test/test_bigmem.py (TODO: 5) dependencies: dependent tests: (no tests depend on bigmem) [x] test: cpython/Lib/test/test_exceptions.py (TODO: 24) dependencies: dependent tests: (no tests depend on exception) [x] lib: cpython/Lib/zipimport.py dependencies:
dependent tests: (9 tests)
[x] test: cpython/Lib/test/test_dtrace.py dependencies: dependent tests: (no tests depend on dtrace) Legend:
|
* Add `test_fileutils.py` from 3.14.6 * Align patch for `test_zipimport_support.py` * Update `test_bigmem.py` to 3.14.6 * Update `test_exception_group.py` * Align patches for `test_baseexception.py` * Update `test_exceptions.py` * Update `test_dtrace.py` * Remove patch from `string_tests.py`
Summary